Skip to content

fix(deps): update module sigs.k8s.io/controller-runtime to v0.25.2 - #141

Merged
cert-manager-prow[bot] merged 1 commit into
mainfrom
renovate/kubernetes-go-patches
Oct 4, 2026
Merged

cert-manager-prow[bot] merged 1 commit into
mainfrom
renovate/kubernetes-go-patches

Conversation

@renovate

@renovate renovate Bot commented Oct 4, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Adoption Passing Confidence
sigs.k8s.io/controller-runtime v0.25.1 → v0.25.2 age adoption passing confidence

Release Notes

kubernetes-sigs/controller-runtime (sigs.k8s.io/controller-runtime)

v0.25.2

Compare Source

What's Changed

Full Changelog: kubernetes-sigs/controller-runtime@v0.25.1...v0.25.2


Configuration

📅 Schedule: (in timezone Europe/London)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
@renovate renovate Bot added dependencies Pull requests that update a dependency file ok-to-test skip-review labels Oct 4, 2026
@cert-manager-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign munnerz for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@cert-manager-prow
cert-manager-prow Bot merged commit 9a2f65e into main Oct 4, 2026
4 checks passed
@renovate
renovate Bot deleted the renovate/kubernetes-go-patches branch October 4, 2026 20:53
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants