Skip to content

fix(deps): update module k8s.io/apimachinery to v0.37.1 - #138

Merged
cert-manager-prow[bot] merged 1 commit into
mainfrom
renovate/kubernetes-go-patches
Sep 27, 2026
Merged

cert-manager-prow[bot] merged 1 commit into
mainfrom
renovate/kubernetes-go-patches

Conversation

@renovate

@renovate renovate Bot commented Sep 27, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Adoption Passing Confidence
k8s.io/apimachinery v0.37.0 → v0.37.1 age adoption passing confidence

Release Notes

kubernetes/apimachinery (k8s.io/apimachinery)

v0.37.1

Compare Source


Configuration

📅 Schedule: (in timezone Europe/London)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
@renovate renovate Bot added dependencies Pull requests that update a dependency file ok-to-test skip-review labels Sep 27, 2026
@cert-manager-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign hjoshi123 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@cert-manager-prow
cert-manager-prow Bot merged commit bd79853 into main Sep 27, 2026
4 checks passed
@renovate
renovate Bot deleted the renovate/kubernetes-go-patches branch September 27, 2026 00:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants