Skip to content

Update CODEOWNERS to AllenCell team - #310

Merged
ShrimpCryptid merged 2 commits into
mainfrom
admin/update-codeowners
Sep 18, 2026
Merged

ShrimpCryptid merged 2 commits into
mainfrom
admin/update-codeowners

Conversation

@jessicasyu

Copy link
Copy Markdown
Contributor

Problem

The current CODEOWNERS points to a team that existed in previous organization. When this repo was migrated to AllenCell, it no longer had access to this team, and therefore the CODEOWNERS file could not assign reviewers.

Solution

Added a new team AllenCell/simviz-web-devs with membership matching the original team, and updated CODEOWNERS to point to this team.

Type of change

  • Bug fix (non-breaking change which fixes an issue)

@github-actions

github-actions Bot commented Sep 17, 2026 •

Copy link
Copy Markdown
PR Preview Action v1.6.3
Preview removed because the pull request was closed.
2026-09-18 23:50 UTC

@ShrimpCryptid
ShrimpCryptid merged commit 2b26c4c into main Sep 18, 2026
2 checks passed
@ShrimpCryptid
ShrimpCryptid deleted the admin/update-codeowners branch September 18, 2026 23:48
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants